⬆️ Update vitest monorepo to ^4.0.8 - #861
Merged
Merged
Conversation
Contributor
Bundle StatsHey there, this message comes from a GitHub action that helps you and reviewers to understand how these changes affect the size of this project's bundle. As this PR is updated, I'll keep you updated on how the bundle size is impacted. Total
Changeset No files were changed View detailed bundle breakdownAdded
Removed
Bigger No assets were bigger Smaller No assets were smaller |
renovate
Bot
force-pushed
the
renovate/vitest-monorepo
branch
from
November 10, 2025 02:33
e1f3bf3 to
65b5534
Compare
renovate
Bot
force-pushed
the
renovate/vitest-monorepo
branch
from
November 10, 2025 02:53
65b5534 to
e292158
Compare
renovate
Bot
force-pushed
the
renovate/vitest-monorepo
branch
from
November 10, 2025 03:15
e292158 to
38ea719
Compare
renovate
Bot
force-pushed
the
renovate/vitest-monorepo
branch
from
November 10, 2025 03:17
38ea719 to
8a3aa43
Compare
renovate
Bot
force-pushed
the
renovate/vitest-monorepo
branch
from
November 10, 2025 03:39
8a3aa43 to
5dc3f75
Compare
renovate
Bot
force-pushed
the
renovate/vitest-monorepo
branch
from
November 10, 2025 03:56
5dc3f75 to
cd2c698
Compare
renovate Bot
added a commit
that referenced
this pull request
Aug 31, 2026
##### [v5.11.0](https://github.com/NaturalIntelligence/fast-xml-parser/releases/tag/v5.11.0) #### What's Changed - add support for endIndex in node metadata (5.x edition) by [@Wain-PC](https://github.com/Wain-PC) in [#850](NaturalIntelligence/fast-xml-parser#850) - fix: don't crash on a closing tag with no matching opening tag by [@hdimer](https://github.com/hdimer) in [#861](NaturalIntelligence/fast-xml-parser#861) #### New Contributors - [@Wain-PC](https://github.com/Wain-PC) made their first contribution in [#850](NaturalIntelligence/fast-xml-parser#850) - [@hdimer](https://github.com/hdimer) made their first contribution in [#861](NaturalIntelligence/fast-xml-parser#861) **Full Changelog**: <NaturalIntelligence/fast-xml-parser@v5.10.1...v5.11.0>
gwennlbh
pushed a commit
that referenced
this pull request
Aug 31, 2026
##### [v5.11.0](https://github.com/NaturalIntelligence/fast-xml-parser/releases/tag/v5.11.0) #### What's Changed - add support for endIndex in node metadata (5.x edition) by [@Wain-PC](https://github.com/Wain-PC) in [#850](NaturalIntelligence/fast-xml-parser#850) - fix: don't crash on a closing tag with no matching opening tag by [@hdimer](https://github.com/hdimer) in [#861](NaturalIntelligence/fast-xml-parser#861) #### New Contributors - [@Wain-PC](https://github.com/Wain-PC) made their first contribution in [#850](NaturalIntelligence/fast-xml-parser#850) - [@hdimer](https://github.com/hdimer) made their first contribution in [#861](NaturalIntelligence/fast-xml-parser#861) **Full Changelog**: <NaturalIntelligence/fast-xml-parser@v5.10.1...v5.11.0> Co-authored-by: renovate[bot] <29139614+renovate[bot]@users.noreply.github.com>
This file contains hidden or bidirectional Unicode text that may be interpreted or compiled differently than what appears below. To review, open the file in an editor that reveals hidden Unicode characters.
Learn more about bidirectional Unicode characters
Sign up for free
to join this conversation on GitHub.
Already have an account?
Sign in to comment
Add this suggestion to a batch that can be applied as a single commit.This suggestion is invalid because no changes were made to the code.Suggestions cannot be applied while the pull request is closed.Suggestions cannot be applied while viewing a subset of changes.Only one suggestion per line can be applied in a batch.Add this suggestion to a batch that can be applied as a single commit.Applying suggestions on deleted lines is not supported.You must change the existing code in this line in order to create a valid suggestion.Outdated suggestions cannot be applied.This suggestion has been applied or marked resolved.Suggestions cannot be applied from pending reviews.Suggestions cannot be applied on multi-line comments.Suggestions cannot be applied while the pull request is queued to merge.Suggestion cannot be applied right now. Please check back later.
This PR contains the following updates:
^4.0.6->^4.0.8^4.0.6->^4.0.8^4.0.6->^4.0.8Release Notes
vitest-dev/vitest (@vitest/coverage-v8)
v4.0.8Compare Source
🐞 Bug Fixes
noExternalmerging bug on Vite 6 - by @hi-ogawa in #8950 (bcb13)expect.element()- by @StyleShit in #8954 (9638d)vi.fn(), fix types forvi.spyOn(obj, class)- by @sheremet-va in #8956 (75e7f)View changes on GitHub
v4.0.7Compare Source
🐞 Bug Fixes
processin case global is overwritten - by @AriPerkkio in #8916 (6240d)isolate: false- by @sheremet-va in #8915 (c9078)toContainElement()matcher - by @vitalybaev in #8910 and #8927 (35a27)isolateoption, deprecatebrowser.isolate/browser.fileParallelism- by @sheremet-va in #8890 (9d2b4)--execArgvas array - by @AriPerkkio in #8924 (751c3)URL.createObjectURL,FormData.set(prop, blob)- by @sheremet-va in #8935 (a1b73)--requireargument when running in deno - by @pi0 in #8897 (d41fa)tsc- by @AriPerkkio in #8920 (fdb2e)🏎 Performance
Array.fromcall - by @Connormiha in #8907 (b6014)View changes on GitHub
Configuration
📅 Schedule: Branch creation - Between 12:00 AM and 04:59 PM, only on Monday ( * 0-16 * * 1 ) (UTC), Automerge - At any time (no schedule defined).
🚦 Automerge: Enabled.
♻ Rebasing: Whenever PR is behind base branch, or you tick the rebase/retry checkbox.
🔕 Ignore: Close this PR and you won't be reminded about these updates again.
This PR was generated by Mend Renovate. View the repository job log.